Skip to content

chore: test shoebox event stream destination - #4824

Closed
Ashish (ashishmax31) wants to merge 1 commit into
mainfrom
test-event-stream-shoebox
Closed

chore: test shoebox event stream destination#4824
Ashish (ashishmax31) wants to merge 1 commit into
mainfrom
test-event-stream-shoebox

Conversation

@ashishmax31

Copy link
Copy Markdown
Collaborator

What

Why

Testing

Special notes for your reviewer

Signed-off-by: Ashish <asnaraya@redhat.com>
@openshift-ci

openshift-ci Bot commented Apr 9, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: ashishmax31
Once this PR has been reviewed and has the lgtm label, please assign bennerv for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Apr 9, 2026

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@ashishmax31

Copy link
Copy Markdown
Collaborator Author

/test stage-e2e-parallel

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant