feat(indexd): improve value display and copy actions - #1304
Merged
alexfreska merged 1 commit intoSep 3, 2025
Conversation
🦋 Changeset detectedLatest commit: d7e2e4f The changes in this PR will be included in the next version bump. This PR includes changesets to release 2 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
|
The latest updates on your projects. Learn more about Vercel for GitHub.
|
This was referenced Sep 3, 2025
Member
Author
This stack of pull requests is managed by Graphite. Learn more about stacking. |
alexfreska
force-pushed
the
feat_indexd_improve_value_display_and_copy_actions
branch
from
September 3, 2025 15:21
f853e94 to
38adac0
Compare
alexfreska
force-pushed
the
fix_indexd_data_explorer_layout_tweaks
branch
from
September 3, 2025 15:21
955d909 to
afa6f0b
Compare
telestrial
approved these changes
Sep 3, 2025
Member
Author
Merge activity
|
alexfreska
changed the base branch from
fix_indexd_data_explorer_layout_tweaks
to
graphite-base/1304
September 3, 2025 19:05
alexfreska
force-pushed
the
feat_indexd_improve_value_display_and_copy_actions
branch
from
September 3, 2025 19:10
38adac0 to
d7e2e4f
Compare
alexfreska
deleted the
feat_indexd_improve_value_display_and_copy_actions
branch
September 3, 2025 19:15
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.

Uh oh!
There was an error while loading. Please reload this page.