Skip to content

chore(deps): bump the dependencies group across 1 directory with 2 updates - #1316

Closed
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/go_modules/sdk/dependencies-aa190b8abf
Closed

chore(deps): bump the dependencies group across 1 directory with 2 updates#1316
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/go_modules/sdk/dependencies-aa190b8abf

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 18, 2025

Copy link
Copy Markdown
Contributor

Bumps the dependencies group with 2 updates in the /sdk directory: go.sia.tech/core and go.sia.tech/coreutils.

Updates go.sia.tech/core from 0.14.1 to 0.17.5

Release notes

Sourced from go.sia.tech/core's releases.

0.17.5 (2025-09-01)

Fixes

  • Fix ProtocolVersion marshalling failing.

0.17.4 (2025-08-27)

Fixes

  • Add compatibility for old protocol version JSON encoding

0.17.3 (2025-08-21)

Fixes

  • Fix RPCSectorRoots not supporting MaxSectorBatchSize.

0.17.2 (2025-08-21)

Fixes

  • Fix RPCAppendSectors not allowing up to MaxSectorBatchSize sectors.

0.17.1 (2025-08-18)

Features

  • Add NonceFactor field to the HardforkASIC struct and use this as the return value for State.NonceFactor().

0.17.0 (2025-08-11)

Breaking Changes

  • Add dedicated ProtocolVersion type for RHP version.

Fixes

  • Increase RPCFreeSectorsRequest and RPCFreeSectorsResponse maxLengths to fix issue with large contracts failing to prune.

0.16.0 (2025-07-28)

Breaking Changes

Allow empty miner payout after FinalCut

A remnant of the old v1 types.Currency encoding persists in the Block.MinerPayouts field. In v2, only a single miner payout is allowed, so its value is fully determined by the sum of the block reward and miner fees. We can therefore omit the value entirely, eliminating a redundant source of truth. It was decided that requiring the value to be omitted was too onerous, as it would compel all miners to update. Making the omission optional allows us to enforce it as a hard requirement at some later time, after all miners have updated.

Add (State).PowTarget and deprecate v1 fields

The consensus.State type contains several redundant and unused PoW-related fields. However, since these fields were included in the Commitment hash, nodes still had to update them correctly. These fields are now zeroed after the hardfork height, so that the Commitment hash can hard-code them if desired.

Features

  • Expose helper methods to compute the remaining allowance, collateral, risked collateral and risked revenue on the V2FileContract type.

Fixes

... (truncated)

Changelog

Sourced from go.sia.tech/core's changelog.

0.17.5 (2025-09-01)

Fixes

  • Fix ProtocolVersion marshalling failing.

0.17.4 (2025-08-27)

Fixes

  • Add compatibility for old protocol version JSON encoding

0.17.3 (2025-08-21)

Fixes

  • Fix RPCSectorRoots not supporting MaxSectorBatchSize.

0.17.2 (2025-08-21)

Fixes

  • Fix RPCAppendSectors not allowing up to MaxSectorBatchSize sectors.

0.17.1 (2025-08-18)

Features

  • Add NonceFactor field to the HardforkASIC struct and use this as the return value for State.NonceFactor().

0.17.0 (2025-08-11)

Breaking Changes

  • Add dedicated ProtocolVersion type for RHP version.

Fixes

  • Increase RPCFreeSectorsRequest and RPCFreeSectorsResponse maxLengths to fix issue with large contracts failing to prune.

0.16.0 (2025-07-28)

Breaking Changes

Add FinalCut difficulty adjustment

The v2 difficulty adjustment algorithm is buggy, leading to higher variance in observed block times. We've fixed the algorithm and confirmed that it exhibits the correct behavior in simulations.

Allow empty miner payout after FinalCut

... (truncated)

Commits
  • b5f2c0a Merge pull request #358 from SiaFoundation/release
  • 111cb97 chore: prepare release 0.17.5
  • a587dad Merge pull request #357 from SiaFoundation/chris/fix-protocol-version-unmarsh...
  • e1ecb33 changeset
  • 637943e fix protocol version unmarshalling
  • a730083 Merge pull request #356 from SiaFoundation/release
  • 32a9054 chore: prepare release 0.17.4
  • d5b75de add compat when unmarshalling protocol
  • 374aacc Merge pull request #354 from SiaFoundation/release
  • 887a65f chore: prepare release 0.17.3
  • Additional commits viewable in compare view

Updates go.sia.tech/coreutils from 0.16.4 to 0.18.4

Release notes

Sourced from go.sia.tech/coreutils's releases.

0.18.4 (2025-09-05)

Features

  • Update protocolVersion to v5.0.1.

0.18.3 (2025-09-02)

Fixes

  • Update core dependency from v0.17.3 to v0.17.5.

0.18.2 (2025-08-26)

Features

  • Make RPCTimeout configurable

Fixes

  • Set keepalive period for siamux tcp connections.

0.18.1 (2025-08-21)

Fixes

  • Fix RPCAppendSectors and RPCSectorRoots not supporting MaxSectorBatchSize.

0.18.0 (2025-08-18)

Breaking Changes

  • Reset Zen testnet.

Features

  • Add ProtocolVersion constants to rhp package.
  • Specify nonce factors for Mainnet and Zen.

Fixes

  • Display migration progress more human-friendly.
  • Increase default RHP4 RPC timeout to 3 minutes.

0.17.0 (2025-07-29)

Breaking Changes

  • Add method to get individual wallet event to SingleAddressWallet
  • Removed unsupported testnets.
  • Removed unused syncer interface from RHP4 server.

Added RPCRefreshPartialRollover

Previously, renters and hosts had to rollover all funds when refreshing a contract. With this change, renters can set their spendable allowance without forcing an increase and hosts only have to rollover their existing risked collateral and revenue. This change increases efficiency of refreshing without compromising the collateral guarantees of existing data or forcing hosts to lock additional collateral that won't be utilized.

... (truncated)

Changelog

Sourced from go.sia.tech/coreutils's changelog.

0.18.4 (2025-09-05)

Features

  • Update protocolVersion to v5.0.1.

0.18.3 (2025-09-02)

Fixes

  • Update core dependency from v0.17.3 to v0.17.5.

0.18.2 (2025-08-26)

Features

  • Make RPCTimeout configurable

Fixes

  • Set keepalive period for siamux tcp connections.

0.18.1 (2025-08-21)

Fixes

  • Fix RPCAppendSectors and RPCSectorRoots not supporting MaxSectorBatchSize.

0.18.0 (2025-08-18)

Breaking Changes

  • Reset Zen testnet.

Features

  • Add ProtocolVersion constants to rhp package.
  • Specify nonce factors for Mainnet and Zen.

Fixes

  • Display migration progress more human-friendly.
  • Increase default RHP4 RPC timeout to 3 minutes.

0.17.0 (2025-07-29)

Breaking Changes

  • Add method to get individual wallet event to SingleAddressWallet
  • Removed unsupported testnets.

... (truncated)

Commits
  • cda2d25 Merge pull request #330 from SiaFoundation/release
  • 578554d chore: prepare release 0.18.4
  • 1c47478 Merge pull request #329 from SiaFoundation/chris/protocol-version
  • 67cbcef changeset
  • 8099b61 update protocol version to 5.0.1
  • 9d582f6 Merge pull request #328 from SiaFoundation/release
  • ccbe3ad chore: prepare release 0.18.3
  • 61c235b update changelog
  • 9c7c743 build(deps): bump go.sia.tech/core from 0.17.3 to 0.17.5
  • 4f261e6 Merge pull request #322 from SiaFoundation/release
  • Additional commits viewable in compare view

Updates go.sia.tech/core from 0.14.1 to 0.17.5

Release notes

Sourced from go.sia.tech/core's releases.

0.17.5 (2025-09-01)

Fixes

  • Fix ProtocolVersion marshalling failing.

0.17.4 (2025-08-27)

Fixes

  • Add compatibility for old protocol version JSON encoding

0.17.3 (2025-08-21)

Fixes

  • Fix RPCSectorRoots not supporting MaxSectorBatchSize.

0.17.2 (2025-08-21)

Fixes

  • Fix RPCAppendSectors not allowing up to MaxSectorBatchSize sectors.

0.17.1 (2025-08-18)

Features

  • Add NonceFactor field to the HardforkASIC struct and use this as the return value for State.NonceFactor().

0.17.0 (2025-08-11)

Breaking Changes

  • Add dedicated ProtocolVersion type for RHP version.

Fixes

  • Increase RPCFreeSectorsRequest and RPCFreeSectorsResponse maxLengths to fix issue with large contracts failing to prune.

0.16.0 (2025-07-28)

Breaking Changes

Allow empty miner payout after FinalCut

A remnant of the old v1 types.Currency encoding persists in the Block.MinerPayouts field. In v2, only a single miner payout is allowed, so its value is fully determined by the sum of the block reward and miner fees. We can therefore omit the value entirely, eliminating a redundant source of truth. It was decided that requiring the value to be omitted was too onerous, as it would compel all miners to update. Making the omission optional allows us to enforce it as a hard requirement at some later time, after all miners have updated.

Add (State).PowTarget and deprecate v1 fields

The consensus.State type contains several redundant and unused PoW-related fields. However, since these fields were included in the Commitment hash, nodes still had to update them correctly. These fields are now zeroed after the hardfork height, so that the Commitment hash can hard-code them if desired.

Features

  • Expose helper methods to compute the remaining allowance, collateral, risked collateral and risked revenue on the V2FileContract type.

Fixes

... (truncated)

Changelog

Sourced from go.sia.tech/core's changelog.

0.17.5 (2025-09-01)

Fixes

  • Fix ProtocolVersion marshalling failing.

0.17.4 (2025-08-27)

Fixes

  • Add compatibility for old protocol version JSON encoding

0.17.3 (2025-08-21)

Fixes

  • Fix RPCSectorRoots not supporting MaxSectorBatchSize.

0.17.2 (2025-08-21)

Fixes

  • Fix RPCAppendSectors not allowing up to MaxSectorBatchSize sectors.

0.17.1 (2025-08-18)

Features

  • Add NonceFactor field to the HardforkASIC struct and use this as the return value for State.NonceFactor().

0.17.0 (2025-08-11)

Breaking Changes

  • Add dedicated ProtocolVersion type for RHP version.

Fixes

  • Increase RPCFreeSectorsRequest and RPCFreeSectorsResponse maxLengths to fix issue with large contracts failing to prune.

0.16.0 (2025-07-28)

Breaking Changes

Add FinalCut difficulty adjustment

The v2 difficulty adjustment algorithm is buggy, leading to higher variance in observed block times. We've fixed the algorithm and confirmed that it exhibits the correct behavior in simulations.

Allow empty miner payout after FinalCut

... (truncated)

Commits
  • b5f2c0a Merge pull request #358 from SiaFoundation/release
  • 111cb97 chore: prepare release 0.17.5
  • a587dad Merge pull request #357 from SiaFoundation/chris/fix-protocol-version-unmarsh...
  • e1ecb33 changeset
  • 637943e fix protocol version unmarshalling
  • a730083 Merge pull request #356 from SiaFoundation/release
  • 32a9054 chore: prepare release 0.17.4
  • d5b75de add compat when unmarshalling protocol
  • 374aacc Merge pull request #354 from SiaFoundation/release
  • 887a65f chore: prepare release 0.17.3
  • Additional commits viewable in compare view

Updates go.sia.tech/coreutils from 0.16.4 to 0.18.4

Release notes

Sourced from go.sia.tech/coreutils's releases.

0.18.4 (2025-09-05)

Features

  • Update protocolVersion to v5.0.1.

0.18.3 (2025-09-02)

Fixes

  • Update core dependency from v0.17.3 to v0.17.5.

0.18.2 (2025-08-26)

Features

  • Make RPCTimeout configurable

Fixes

  • Set keepalive period for siamux tcp connections.

0.18.1 (2025-08-21)

Fixes

  • Fix RPCAppendSectors and RPCSectorRoots not supporting MaxSectorBatchSize.

0.18.0 (2025-08-18)

Breaking Changes

  • Reset Zen testnet.

Features

  • Add ProtocolVersion constants to rhp package.
  • Specify nonce factors for Mainnet and Zen.

Fixes

  • Display migration progress more human-friendly.
  • Increase default RHP4 RPC timeout to 3 minutes.

0.17.0 (2025-07-29)

Breaking Changes

  • Add method to get individual wallet event to SingleAddressWallet
  • Removed unsupported testnets.
  • Removed unused syncer interface from RHP4 server.

Added RPCRefreshPartialRollover

Previously, renters and hosts had to rollover all funds when refreshing a contract. With this change, renters can set their spendable allowance without forcing an increase and hosts only have to rollover their existing risked collateral and revenue. This change increases efficiency of refreshing without compromising the collateral guarantees of existing data or forcing hosts to lock additional collateral that won't be utilized.

... (truncated)

Changelog

Sourced from go.sia.tech/coreutils's changelog.

0.18.4 (2025-09-05)

Features

  • Update protocolVersion to v5.0.1.

0.18.3 (2025-09-02)

Fixes

  • Update core dependency from v0.17.3 to v0.17.5.

0.18.2 (2025-08-26)

Features

  • Make RPCTimeout configurable

Fixes

  • Set keepalive period for siamux tcp connections.

0.18.1 (2025-08-21)

Fixes

  • Fix RPCAppendSectors and RPCSectorRoots not supporting MaxSectorBatchSize.

0.18.0 (2025-08-18)

Breaking Changes

  • Reset Zen testnet.

Features

  • Add ProtocolVersion constants to rhp package.
  • Specify nonce factors for Mainnet and Zen.

Fixes

  • Display migration progress more human-friendly.
  • Increase default RHP4 RPC timeout to 3 minutes.

0.17.0 (2025-07-29)

Breaking Changes

  • Add method to get individual wallet event to SingleAddressWallet
  • Removed unsupported testnets.

... (truncated)

Commits
  • cda2d25 Merge pull request #330 from SiaFoundation/release
  • 578554d chore: prepare release 0.18.4
  • 1c47478 Merge pull request #329 from SiaFoundation/chris/protocol-version
  • 67cbcef changeset
  • 8099b61 update protocol version to 5.0.1
  • 9d582f6 Merge pull request #328 from SiaFoundation/release
  • ccbe3ad chore: prepare release 0.18.3
  • 61c235b update changelog
  • 9c7c743 build(deps): bump go.sia.tech/core from 0.17.3 to 0.17.5
  • 4f261e6 Merge pull request #322 from SiaFoundation/release
  • Additional commits viewable in compare view

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot merge will merge this PR after your CI passes on it
  • @dependabot squash and merge will squash and merge this PR after your CI passes on it
  • @dependabot cancel merge will cancel a previously requested merge and block automerging
  • @dependabot reopen will reopen this PR if it is closed
  • @dependabot close will close this PR and stop Dependabot recreating it. You can achieve the same result by closing it manually
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore <dependency name> major version will close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)
  • @dependabot ignore <dependency name> minor version will close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)
  • @dependabot ignore <dependency name> will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)
  • @dependabot unignore <dependency name> will remove all of the ignore conditions of the specified dependency
  • @dependabot unignore <dependency name> <ignore condition> will remove the ignore condition of the specified dependency and ignore conditions

…dates

Bumps the dependencies group with 2 updates in the /sdk directory: [go.sia.tech/core](https://github.com/SiaFoundation/core) and [go.sia.tech/coreutils](https://github.com/SiaFoundation/coreutils).


Updates `go.sia.tech/core` from 0.14.1 to 0.17.5
- [Release notes](https://github.com/SiaFoundation/core/releases)
- [Changelog](https://github.com/SiaFoundation/core/blob/master/CHANGELOG.md)
- [Commits](SiaFoundation/core@v0.14.1...v0.17.5)

Updates `go.sia.tech/coreutils` from 0.16.4 to 0.18.4
- [Release notes](https://github.com/SiaFoundation/coreutils/releases)
- [Changelog](https://github.com/SiaFoundation/coreutils/blob/master/CHANGELOG.md)
- [Commits](SiaFoundation/coreutils@v0.16.4...v0.18.4)

Updates `go.sia.tech/core` from 0.14.1 to 0.17.5
- [Release notes](https://github.com/SiaFoundation/core/releases)
- [Changelog](https://github.com/SiaFoundation/core/blob/master/CHANGELOG.md)
- [Commits](SiaFoundation/core@v0.14.1...v0.17.5)

Updates `go.sia.tech/coreutils` from 0.16.4 to 0.18.4
- [Release notes](https://github.com/SiaFoundation/coreutils/releases)
- [Changelog](https://github.com/SiaFoundation/coreutils/blob/master/CHANGELOG.md)
- [Commits](SiaFoundation/coreutils@v0.16.4...v0.18.4)

---
updated-dependencies:
- dependency-name: go.sia.tech/core
  dependency-version: 0.17.5
  dependency-type: direct:production
  update-type: version-update:semver-minor
  dependency-group: dependencies
- dependency-name: go.sia.tech/coreutils
  dependency-version: 0.18.4
  dependency-type: direct:production
  update-type: version-update:semver-minor
  dependency-group: dependencies
- dependency-name: go.sia.tech/core
  dependency-version: 0.17.5
  dependency-type: direct:production
  update-type: version-update:semver-minor
  dependency-group: dependencies
- dependency-name: go.sia.tech/coreutils
  dependency-version: 0.18.4
  dependency-type: direct:production
  update-type: version-update:semver-minor
  dependency-group: dependencies
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file go Pull requests that update Go code labels Sep 18, 2025
@vercel

vercel Bot commented Sep 18, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

5 Skipped Deployments
Project Deployment Preview Comments Updated (UTC)
explorer Ignored Ignored Preview Sep 18, 2025 3:00pm
explorer-zen Ignored Ignored Preview Sep 18, 2025 3:00pm
hostd Ignored Ignored Preview Sep 18, 2025 3:00pm
indexd Ignored Ignored Preview Sep 18, 2025 3:00pm
renterd Ignored Ignored Preview Sep 18, 2025 3:00pm

@changeset-bot

changeset-bot Bot commented Sep 18, 2025

Copy link
Copy Markdown

⚠️ No Changeset found

Latest commit: 287e5eb

Merging this PR will not cause a version bump for any packages. If these changes should not result in a new version, you're good to go. If these changes should result in a version bump, you need to add a changeset.

This PR includes no changesets

When changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types

Click here to learn what changesets are, and how to add one.

Click here if you're a maintainer who wants to add a changeset to this PR

@dependabot @github

dependabot Bot commented on behalf of github Sep 25, 2025

Copy link
Copy Markdown
Contributor Author

Superseded by #1332.

@dependabot dependabot Bot closed this Sep 25, 2025
@dependabot
dependabot Bot deleted the dependabot/go_modules/sdk/dependencies-aa190b8abf branch September 25, 2025 09:25
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file go Pull requests that update Go code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants