Skip to content

chore: Move CODEOWNERS - #43

Closed
fxgst wants to merge 1 commit into
mainfrom
elias/move
Closed

chore: Move CODEOWNERS#43
fxgst wants to merge 1 commit into
mainfrom
elias/move

Conversation

@fxgst

@fxgst fxgst commented Aug 28, 2025

Copy link
Copy Markdown
Contributor

No description provided.

@fxgst
fxgst requested a review from a team as a code owner August 28, 2025 12:06
@fxgst fxgst closed this Sep 11, 2025
@fxgst
fxgst deleted the elias/move branch September 11, 2025 08:11
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant