Skip to content

chore: add CODEOWNERS file for customer success team - #78

Merged
mbevc1 merged 2 commits into
mainfrom
mbevc1-patch-1
Mar 18, 2026
Merged

chore: add CODEOWNERS file for customer success team#78
mbevc1 merged 2 commits into
mainfrom
mbevc1-patch-1

Conversation

@mbevc1

@mbevc1 mbevc1 commented Mar 17, 2026

Copy link
Copy Markdown
Contributor

Add CODEOWNERS for docs and notify reviewers. Needs amending permissions and add CS team group as Admin or at least writer.

@mbevc1
mbevc1 marked this pull request as ready for review March 17, 2026 11:46
@mbevc1 mbevc1 changed the title chore: Add CODEOWNERS file for customer success team chore: add CODEOWNERS file for customer success team Mar 17, 2026
@mbevc1
mbevc1 enabled auto-merge (squash) March 17, 2026 12:02
@mbevc1
mbevc1 requested a review from dangrondahl March 17, 2026 16:26
@mbevc1
mbevc1 merged commit 23b06bb into main Mar 18, 2026
5 of 6 checks passed
@mbevc1
mbevc1 deleted the mbevc1-patch-1 branch March 18, 2026 12:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants