Skip to content

en,zh: remove TidbClusterAutoScaler references as it is deprecated (#3105) - #3111

Merged
ti-chi-bot[bot] merged 2 commits into
pingcap:release-1.5from
ti-chi-bot:cherry-pick-3105-to-release-1.5
Mar 3, 2026
Merged

en,zh: remove TidbClusterAutoScaler references as it is deprecated (#3105)#3111
ti-chi-bot[bot] merged 2 commits into
pingcap:release-1.5from
ti-chi-bot:cherry-pick-3105-to-release-1.5

Conversation

@ti-chi-bot

Copy link
Copy Markdown
Member

This is an automated cherry-pick of #3105

First-time contributors' checklist

What is changed, added, or deleted? (Required)

Which TiDB Operator version(s) do your changes apply to? (Required)

  • main (the latest development version for v2.x)
  • v2.0 (TiDB Operator 2.0 versions)
  • release-1.x (the latest development version for v1.x)
  • v1.6 (TiDB Operator 1.6 versions)
  • v1.5 (TiDB Operator 1.5 versions)
  • v1.4 (TiDB Operator 1.4 versions)
  • v1.3 (TiDB Operator 1.3 versions)

What is the related PR or file link(s)?

  • This PR is translated from:
  • Other reference link(s):

@ti-chi-bot ti-chi-bot added lgtm needs-1-more-lgtm Indicates a PR needs 1 more LGTM. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. type/cherry-pick-for-release-1.5 labels Mar 3, 2026
@Oreoxmt

Oreoxmt commented Mar 3, 2026

Copy link
Copy Markdown
Contributor

/approve

@ti-chi-bot

ti-chi-bot Bot commented Mar 3, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Oreoxmt

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ti-chi-bot ti-chi-bot Bot added the approved label Mar 3, 2026
@ti-chi-bot
ti-chi-bot Bot merged commit c761a51 into pingcap:release-1.5 Mar 3, 2026
13 checks passed
@ti-chi-bot
ti-chi-bot Bot deleted the cherry-pick-3105-to-release-1.5 branch March 3, 2026 08:28
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved lgtm needs-1-more-lgtm Indicates a PR needs 1 more LGTM. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. type/cherry-pick-for-release-1.5

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants