feat(design-system): data table - #1231
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 3cb9287 The changes in this PR will be included in the next version bump. This PR includes changesets to release 2 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
|
The latest updates on your projects. Learn more about Vercel for Git ↗︎
4 Skipped Deployments
|
This was referenced Aug 1, 2025
This was referenced Aug 1, 2025
Member
Author
alexfreska
force-pushed
the
feat_ds_changes
branch
from
August 1, 2025 18:30
f1f4814 to
d091823
Compare
alexfreska
force-pushed
the
feat_data_table
branch
from
August 1, 2025 18:30
3efa8d3 to
f79e68e
Compare
alexfreska
force-pushed
the
feat_data_table
branch
from
August 1, 2025 20:55
f79e68e to
751bf92
Compare
alexfreska
force-pushed
the
feat_ds_changes
branch
from
August 1, 2025 20:55
d091823 to
76e9c68
Compare
alexfreska
force-pushed
the
feat_data_table
branch
from
August 4, 2025 14:22
751bf92 to
ea1b5f3
Compare
alexfreska
force-pushed
the
feat_ds_changes
branch
from
August 4, 2025 14:22
76e9c68 to
0f99dd0
Compare
telestrial
approved these changes
Aug 4, 2025
alexfreska
force-pushed
the
feat_ds_changes
branch
from
August 4, 2025 21:27
0f99dd0 to
ce6dac4
Compare
alexfreska
force-pushed
the
feat_data_table
branch
from
August 4, 2025 21:27
ea1b5f3 to
1b44b27
Compare
alexfreska
force-pushed
the
feat_ds_changes
branch
from
August 4, 2025 21:32
ce6dac4 to
e7d26f8
Compare
alexfreska
force-pushed
the
feat_data_table
branch
from
August 4, 2025 21:32
1b44b27 to
20eb1f7
Compare
Member
Author
alexfreska
force-pushed
the
feat_data_table
branch
from
August 4, 2025 22:11
20eb1f7 to
3cb9287
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.

screenshot from next PR which uses it in indexd