Skip to content

feat(indexd): contract explorer - #1232

Merged
alexfreska merged 1 commit into
mainfrom
feat_indexd_contract_explorer
Aug 4, 2025
Merged

feat(indexd): contract explorer#1232
alexfreska merged 1 commit into
mainfrom
feat_indexd_contract_explorer

Conversation

@alexfreska

@alexfreska alexfreska commented Aug 1, 2025

Copy link
Copy Markdown
Member
Screenshot 2025-08-01 at 4 13 48 PM Screenshot 2025-08-01 at 4 15 30 PM

@vercel

vercel Bot commented Aug 1, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
indexd ✅ Ready (Inspect) Visit Preview 💬 Add feedback Aug 4, 2025 10:30pm
4 Skipped Deployments
Name Status Preview Comments Updated (UTC)
explorer ⬜️ Ignored (Inspect) Visit Preview Aug 4, 2025 10:30pm
explorer-zen ⬜️ Ignored (Inspect) Visit Preview Aug 4, 2025 10:30pm
hostd ⬜️ Ignored (Inspect) Visit Preview Aug 4, 2025 10:30pm
renterd ⬜️ Ignored (Inspect) Visit Preview Aug 4, 2025 10:30pm

@changeset-bot

changeset-bot Bot commented Aug 1, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ea27cab

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
indexd Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

alexfreska commented Aug 1, 2025

Copy link
Copy Markdown
Member Author

@alexfreska
alexfreska force-pushed the feat_indexd_host_explorer_filter_sort branch from 95090c2 to d6064a4 Compare August 4, 2025 21:27
@alexfreska
alexfreska force-pushed the feat_indexd_contract_explorer branch from af2ba58 to 1f00936 Compare August 4, 2025 21:27
@alexfreska
alexfreska force-pushed the feat_indexd_host_explorer_filter_sort branch from d6064a4 to f80eed6 Compare August 4, 2025 21:32

alexfreska commented Aug 4, 2025

Copy link
Copy Markdown
Member Author

Merge activity

@alexfreska
alexfreska force-pushed the feat_indexd_host_explorer_filter_sort branch from f80eed6 to 56ff824 Compare August 4, 2025 22:20
@alexfreska
alexfreska force-pushed the feat_indexd_contract_explorer branch from 6ceab1a to 0547716 Compare August 4, 2025 22:20
@alexfreska
alexfreska force-pushed the feat_indexd_host_explorer_filter_sort branch from 56ff824 to b667fd1 Compare August 4, 2025 22:22
@alexfreska
alexfreska force-pushed the feat_indexd_contract_explorer branch from 0547716 to b4a1eeb Compare August 4, 2025 22:22
Base automatically changed from feat_indexd_host_explorer_filter_sort to main August 4, 2025 22:27
@alexfreska
alexfreska force-pushed the feat_indexd_contract_explorer branch from b4a1eeb to ea27cab Compare August 4, 2025 22:27
@alexfreska
alexfreska merged commit 4a76bb7 into main Aug 4, 2025
1 of 65 checks passed
@alexfreska
alexfreska deleted the feat_indexd_contract_explorer branch August 4, 2025 22:27
@github-project-automation github-project-automation Bot moved this from In Progress to Done in Sia Aug 4, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

Archived in project

Development

Successfully merging this pull request may close these issues.

3 participants